Skip to content

chore(deps): bump sass from 1.100.0 to 1.105.0 in /attack-style in the npm-minor-and-patch group across 1 directory - #602

Open
dependabot[bot] wants to merge 1 commit into
developfrom
dependabot/npm_and_yarn/attack-style/develop/npm-minor-and-patch-eca4739428
Open

dependabot[bot] wants to merge 1 commit into
developfrom
dependabot/npm_and_yarn/attack-style/develop/npm-minor-and-patch-eca4739428

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Jun 18, 2026 •

Copy link
Copy Markdown
Contributor

Bumps the npm-minor-and-patch group with 1 update in the /attack-style directory: sass.

Updates sass from 1.100.0 to 1.105.0

Release notes

Sourced from sass's releases.

Dart Sass 1.105.0

To install Sass 1.105.0, download one of the packages below and add it to your PATH, or see the Sass website for full installation instructions.

Changes

  • Add support for first-class modules. These can be accessed using the new meta.load() and meta.get-module() functions, and may be passed as the $module argument to numerous eisting sass:meta functions.

  • Add the meta.css() mixin, which includes CSS from a first-class module.

JS API

  • Add a SassModule class and a corresponding Value.assertModule() method.

Dart API

  • Add a SassModule class and a corresponding Value.assertModule() method.

See the full changelog for changes in earlier releases.

Dart Sass 1.104.1

To install Sass 1.104.1, download one of the packages below and add it to your PATH, or see the Sass website for full installation instructions.

Changes

  • Fix a bug where loud comments before @use rules could be emitted multiple times under certain circumstances.

  • Fix a bug in which an invalid selector at the end of an indented syntax stylesheet would cause the parser to crash instead of emitting a useful error message.

Command Line Interface

  • Many-to-many compilations no longer compile any files that appear in the output directory if the output directory is also within the source directory. This fixes a bug where --watch mode could enter an infinite loop recompiling the same CSS file over and over.

  • Sass now sets the modification time of output files to the time compilation started rather than the time it ended. This ensures that, if a source file is modified during compilation, --watch and --update mode will recompile the outputs to include the new source file contents.

See the full changelog for changes in earlier releases.

Dart Sass 1.104.0

To install Sass 1.104.0, download one of the packages below and add it to your PATH, or see the Sass website for full installation instructions.

Changes

... (truncated)

Changelog

Sourced from sass's changelog.

1.105.0

  • Add support for first-class modules. These can be accessed using the new meta.load() and meta.get-module() functions, and may be passed as the $module argument to numerous eisting sass:meta functions.

  • Add the meta.css() mixin, which includes CSS from a first-class module.

JS API

  • Add a SassModule class and a corresponding Value.assertModule() method.

Dart API

  • Add a SassModule class and a corresponding Value.assertModule() method.

1.104.1

  • Fix a bug where loud comments before @use rules could be emitted multiple times under certain circumstances.

  • Fix a bug in which an invalid selector at the end of an indented syntax stylesheet would cause the parser to crash instead of emitting a useful error message.

Command Line Interface

  • Many-to-many compilations no longer compile any files that appear in the output directory if the output directory is also within the source directory. This fixes a bug where --watch mode could enter an infinite loop recompiling the same CSS file over and over.

  • Sass now sets the modification time of output files to the time compilation started rather than the time it ended. This ensures that, if a source file is modified during compilation, --watch and --update mode will recompile the outputs to include the new source file contents.

1.104.0

  • Potentially breaking compatibility fix: Colors now convert the special values NaN and negative zero, as well as infinity and negative infinity for polar-hue channels, to 0 as per the CSS spec.

  • The special value negative zero is now serialized as -0 instead of 0 for greater compatibility when using it in CSS calculations.

1.103.1

  • No user-visible changes.

... (truncated)

Commits
  • 4bf2b92 Bump zizmorcore/zizmor-action from 0.6.3 to 0.6.4 (#2865)
  • 850d57e [meta.load] Add support for meta.load() and related features (#2861)
  • 6180be0 Bump postcss from 8.5.26 to 8.5.28 in /pkg/sass-parser (#2863)
  • da84fb5 Fix --watch not recompiling sources modified during a compilation (#2850)
  • 8a287a7 Merge pull request #2859 from sass/fix-auto-merge
  • 6033b46 Add explicit versions to .github/util actions
  • cc71d43 Only auto-merge dependabot PRs once the tests pass
  • 49ddc4f Revert "Bump dart-lang/setup-dart in /.github/util/initialize"
  • 71a127d Avoid duplicating loud comments that appear before @use (#2854)
  • f19baab Merge pull request #2857 from sass/dependabot/github_actions/dot-github/util/...
  • Additional commits viewable in compare view

@dependabot dependabot Bot added dependencies Pull requests that update a dependency file javascript Pull requests that update Javascript code labels Jun 18, 2026
@dependabot dependabot Bot changed the title chore(deps): bump sass from 1.100.0 to 1.101.0 in /attack-style in the npm-minor-and-patch group chore(deps): bump sass from 1.100.0 to 1.101.0 in /attack-style in the npm-minor-and-patch group across 1 directory Jun 19, 2026
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/attack-style/develop/npm-minor-and-patch-eca4739428 branch from aa47f05 to 0db2574 Compare June 19, 2026 20:46
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/attack-style/develop/npm-minor-and-patch-eca4739428 branch from 0db2574 to 1245ab4 Compare July 1, 2026 18:58
@sonarqubecloud

sonarqubecloud Bot commented Jul 1, 2026

Copy link
Copy Markdown

@dependabot dependabot Bot changed the title chore(deps): bump sass from 1.100.0 to 1.101.0 in /attack-style in the npm-minor-and-patch group across 1 directory chore(deps): bump sass from 1.100.0 to 1.102.0 in /attack-style in the npm-minor-and-patch group across 1 directory Aug 1, 2026
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/attack-style/develop/npm-minor-and-patch-eca4739428 branch from 1245ab4 to 5065a6a Compare August 1, 2026 18:56
@sonarqubecloud

sonarqubecloud Bot commented Aug 1, 2026

Copy link
Copy Markdown

@dependabot dependabot Bot changed the title chore(deps): bump sass from 1.100.0 to 1.102.0 in /attack-style in the npm-minor-and-patch group across 1 directory chore(deps): bump sass from 1.100.0 to 1.103.1 in /attack-style in the npm-minor-and-patch group across 1 directory Sep 1, 2026
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/attack-style/develop/npm-minor-and-patch-eca4739428 branch from 5065a6a to fe75a76 Compare September 1, 2026 18:57
Bumps the npm-minor-and-patch group with 1 update in the /attack-style directory: [sass](https://github.com/sass/dart-sass).


Updates `sass` from 1.100.0 to 1.105.0
- [Release notes](https://github.com/sass/dart-sass/releases)
- [Changelog](https://github.com/sass/dart-sass/blob/main/CHANGELOG.md)
- [Commits](sass/dart-sass@1.100.0...1.105.0)

---
updated-dependencies:
- dependency-name: sass
  dependency-version: 1.101.0
  dependency-type: direct:production
  update-type: version-update:semver-minor
  dependency-group: npm-minor-and-patch
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot changed the title chore(deps): bump sass from 1.100.0 to 1.103.1 in /attack-style in the npm-minor-and-patch group across 1 directory chore(deps): bump sass from 1.100.0 to 1.105.0 in /attack-style in the npm-minor-and-patch group across 1 directory Oct 1, 2026
@dependabot
dependabot Bot force-pushed the dependabot/npm_and_yarn/attack-style/develop/npm-minor-and-patch-eca4739428 branch from fe75a76 to 40077a9 Compare October 1, 2026 18:59
@sonarqubecloud

sonarqubecloud Bot commented Oct 1, 2026

Copy link
Copy Markdown

This branch has not been deployed

No deployments
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file javascript Pull requests that update Javascript code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants